Skip to content

[vncproxy]Disable traffic logging - #1144

Merged
openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
gibizer:vnc-traffic-log
Jul 7, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
gibizer:vnc-traffic-log

Conversation

@gibizer

@gibizer gibizer commented Jun 29, 2026

Copy link
Copy Markdown
Contributor

By default we set [DEFAULT]debug=true in all service config to increase our troubleshooting chances. But vncproxy with that setting emits a lot of traffic logging. By setting debug=false (and with the nova fix in depends-on) we disable traffic logging by default.

Depends-On: https://gitlab.cee.redhat.com/eng/openstack/nova/-/merge_requests/95
Related: OSPRH-29916

By default we set [DEFAULT]debug=true in all service config to increase
our troubleshooting chances. But vncproxy with that setting emits a lot
of traffic logging. By setting debug=false (and with the nova fix in
depends-on) we disable traffic logging by default.

Depends-On: https://gitlab.cee.redhat.com/eng/openstack/nova/-/merge_requests/95
Related: OSPRH-29916
Signed-off-by: Balazs Gibizer <gibi@redhat.com>
@openshift-ci
openshift-ci Bot requested review from amoralej and jamepark4 June 29, 2026 15:33
@gibizer

gibizer commented Jun 29, 2026

Copy link
Copy Markdown
Contributor Author

/hold waiting for the nova fix to land

@gibizer

gibizer commented Jul 6, 2026

Copy link
Copy Markdown
Contributor Author

/unhold
nova backport landed

@openshift-ci

openshift-ci Bot commented Jul 7, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: gibizer, mrkisaolamb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [gibizer,mrkisaolamb]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 3377647 into openstack-k8s-operators:main Jul 7, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants