Repository navigation
[release-v1.16] Update OWNERS file - #1923
Open
serverless-qe wants to merge 1 commit into
Open
serverless-qe wants to merge 1 commit into
serverless-qe wants to merge 1 commit into
Conversation
Collaborator
|
/lgtm |
serverless-qe
force-pushed
the
sync-owners-release-v1.16
branch
from
September 29, 2026 11:46
340e4f6 to
cd07c78
Compare
Collaborator
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: creydr, serverless-qe The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
serverless-qe
force-pushed
the
sync-owners-release-v1.16
branch
from
September 29, 2026 18:07
cd07c78 to
cd59da7
Compare
|
New changes are detected. LGTM label has been removed. |
serverless-qe
force-pushed
the
sync-owners-release-v1.16
branch
6 times, most recently
from
October 5, 2026 14:10
687f32f to
ff372e5
Compare
serverless-qe
force-pushed
the
sync-owners-release-v1.16
branch
from
October 6, 2026 19:03
ff372e5 to
d0049de
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Update OWNERS file