Update self-hosted to clarify LAUNCHPAD_RPC_SHARED_SECRET - #19346
Merged
Merged
Conversation
Improving the explanation around how the LAUNCHPAD_RPC_SHARED_SECRET is used. This is more accurate on how the secret is shared among services.
|
The latest updates on your projects. Learn more about Vercel for GitHub.
1 Skipped Deployment
|
NicoHinderling
approved these changes
Sep 10, 2026
coolguyzone
reviewed
Sep 14, 2026
| Sentry comes with a cleanup cron job that prunes events older than `90 days` by default. If you want to change that, you can edit the `SENTRY_EVENT_RETENTION_DAYS` environment variable in `.env` or simply override it in your environment. | ||
|
|
||
| ### Internal RPC Shared Secret | ||
| ### Launchpad RPC shared secret |
Contributor
There was a problem hiding this comment.
Suggested change
| ### Launchpad RPC shared secret | |
| ### Launchpad RPC Shared Secret |
coolguyzone
approved these changes
Sep 14, 2026
coolguyzone
left a comment
Contributor
There was a problem hiding this comment.
LGTM, thanks for updating!
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
DESCRIBE YOUR PR
Improving the explanation around how the LAUNCHPAD_RPC_SHARED_SECRET is used. This is more accurate on how the secret is shared among services.
IS YOUR CHANGE URGENT?
Help us prioritize incoming PRs by letting us know when the change needs to go live.
Select exactly one option. For deadlines, replace
YYYY-MM-DDwith the due date. You can update this information later by editing the PR description.SLA
Thanks in advance for your help!
PRE-MERGE CHECKLIST
Make sure you've checked the following before merging your changes: